Skip to content

feat(rust): add wait_for_configuration_timeout() method - #339

Merged
dd-oleksii merged 1 commit into
Eppo-exp:mainfrom
dd-oleksii:feat-rust-wait-for-configuration-timeout
Oct 13, 2025
Merged

dd-oleksii merged 1 commit into
Eppo-exp:mainfrom
dd-oleksii:feat-rust-wait-for-configuration-timeout

Conversation

@dd-oleksii

@dd-oleksii dd-oleksii commented Oct 10, 2025

Copy link
Copy Markdown
Collaborator

Add wait_for_configuration_timeout() method.

In poor network conditions, wait_for_configuration() may block waiting on configuration indefinitely which may be undesired. Add a new wait_for_configuration_timeout() which allows specifying a timeout for waiting.

It used to be hard to implement because we didn't have shared async runtime (that's why we had indefinitely-blocking version initially). But now we have a shared async runtime, so handling timeouts is easier.

@changeset-bot

changeset-bot Bot commented Oct 10, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f67d165

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
Name Type
eppo_core Minor
rust-sdk Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@rasendubi

Copy link
Copy Markdown
Collaborator

The CI error is unrelated to the PR

@leoromanovsky leoromanovsky left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Good idea marking the other function deprecated

Comment thread rust-sdk/src/poller.rs Outdated
@dd-oleksii
dd-oleksii force-pushed the feat-rust-wait-for-configuration-timeout branch from bb8de1f to f67d165 Compare October 13, 2025 11:46
@dd-oleksii
dd-oleksii merged commit 9a4d2a5 into Eppo-exp:main Oct 13, 2025
38 of 45 checks passed
@dd-oleksii
dd-oleksii deleted the feat-rust-wait-for-configuration-timeout branch October 13, 2025 12:30
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants