Skip to content

fscryptctl: init at 1.0.0 - #112651

Merged
primeos merged 1 commit into
NixOS:masterfrom
primeos:fscryptctl
Feb 11, 2021
Merged

fscryptctl: init at 1.0.0#112651
primeos merged 1 commit into
NixOS:masterfrom
primeos:fscryptctl

Conversation

@primeos

@primeos primeos commented Feb 10, 2021

Copy link
Copy Markdown
Member

Release notes: https://github.com/google/fscryptctl/releases/tag/v1.0.0

fscryptctl-experimental will remain at version 0.1.0 to ensure a smooth
transition.

Motivation for this change
Things done
  • Tested using sandboxing (nix.useSandbox on NixOS, or option sandbox in nix.conf on non-NixOS linux)
  • Built on platform(s)
    • NixOS
    • macOS
    • other Linux distributions
  • Tested via one or more NixOS test(s) if existing and applicable for the change (look inside nixos/tests)
  • Tested compilation of all pkgs that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review wip"
  • Tested execution of all binary files (usually in ./result/bin/)
  • Determined the impact on package closure size (by running nix path-info -S before and after)
  • Ensured that relevant documentation is up to date
  • Fits CONTRIBUTING.md.

Release notes: https://github.com/google/fscryptctl/releases/tag/v1.0.0

fscryptctl-experimental will remain at version 0.1.0 to ensure a smooth
transition.
license = licenses.asl20;
platforms = platforms.linux;
maintainers = with maintainers; [ primeos ];
knownVulnerabilities = [ ''

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I used knownVulnerabilities because #109407 isn't ready yet and "the v1 encryption policies can be insecure".
This is a bit of a stretch but IMO it's still the best approach.

@ofborg ofborg Bot added 8.has: package (new) This PR adds a new package 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. labels Feb 10, 2021
@primeos
primeos merged commit 7fdadc1 into NixOS:master Feb 11, 2021
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

8.has: package (new) This PR adds a new package 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant