Skip to content

transmission, transmission_4: set mainProgram and restore nixosTests - #256866

Merged
doronbehar merged 3 commits into
NixOS:masterfrom
kirillrdy:transmission
Oct 4, 2023
Merged

doronbehar merged 3 commits into
NixOS:masterfrom
kirillrdy:transmission

Conversation

@kirillrdy

@kirillrdy kirillrdy commented Sep 23, 2023

Copy link
Copy Markdown
Member

Description of changes

Things done

  • Built on platform(s)
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • For non-Linux: Is sandbox = true set in nix.conf? (See Nix manual)
  • Tested, as applicable:
  • Tested compilation of all packages that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage
  • Tested basic functionality of all binary files (usually in ./result/bin/)
  • 23.11 Release Notes (or backporting 23.05 Release notes)
    • (Package updates) Added a release notes entry if the change is major or breaking
    • (Module updates) Added a release notes entry if the change is significant
    • (Module addition) Added a release notes entry if adding a new NixOS module
  • Fits CONTRIBUTING.md.

Comment thread pkgs/applications/networking/p2p/transmission/default.nix Outdated
@kirillrdy
kirillrdy marked this pull request as ready for review September 23, 2023 10:28
@jtojnar

jtojnar commented Sep 23, 2023

Copy link
Copy Markdown
Member

We already have it since #241146. Also note there are data-loss issues: #215316 (comment)

@kirillrdy

Copy link
Copy Markdown
Member Author

We already have it since #241146. Also note there are data-loss issues: #215316 (comment)

ah thanks, github search :'(

I'll de-scope this to just mainProgram change

@kirillrdy
kirillrdy marked this pull request as draft September 23, 2023 10:47
@ofborg
ofborg Bot requested a review from astsmtl September 23, 2023 10:50
@ofborg ofborg Bot added 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. labels Sep 23, 2023
@kirillrdy kirillrdy changed the title transmission: 3.00 -> 4.0.3 transmission, transmission_4: set mainProgram Sep 23, 2023
@kirillrdy
kirillrdy marked this pull request as ready for review September 23, 2023 10:55
@kirillrdy kirillrdy changed the title transmission, transmission_4: set mainProgram transmission, transmission_4: set mainProgram and restore nixosTests Sep 23, 2023
@kirillrdy
kirillrdy requested a review from jtojnar September 23, 2023 11:01
@ofborg ofborg Bot added 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 0 This PR does not cause any packages to rebuild on Linux. and removed 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. labels Sep 23, 2023
@github-actions github-actions Bot added the 6.topic: nixos Issues or PRs affecting NixOS modules, or package usability issues specific to NixOS label Sep 23, 2023
Comment thread pkgs/applications/networking/p2p/transmission/4.nix Outdated
Comment thread nixos/tests/all-tests.nix
trafficserver = handleTest ./trafficserver.nix {};
transmission = handleTest ./transmission.nix {};
transmission = handleTest ./transmission.nix { transmission = pkgs.transmission; };
transmission_4 = handleTest ./transmission.nix { transmission = pkgs.transmission_4; };

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

i am happy to drop this if people feel like there isn't much value

@delroth delroth added the 12.approvals: 1 This PR was reviewed and approved by one person. label Sep 28, 2023
@doronbehar

Copy link
Copy Markdown
Contributor

You all might be interested in #258058 .


passthru.tests = {
apparmor = nixosTests.transmission_4; # starts the service with apparmor enabled
smoke-test = nixosTests.bittorrent;

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Is not this testing Transmission 3?

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

this is transmission_4, I assume we want to test transmission_4 here

@kirillrdy

Copy link
Copy Markdown
Member Author

You all might be interested in #258058 .

Yes !!! while I know 4 has a regression, I think it will be less confusing and inline with other distributions, to have 4 as default.

I am happy to put this in draft and wait for rename to happen

@jtojnar

jtojnar commented Sep 29, 2023

Copy link
Copy Markdown
Member

I think it will be less confusing and inline with other distributions, to have 4 as default.

It might also delete people’s data unless they are aware of the bug and switch to transmission_3 before running transmission.

@kirillrdy

Copy link
Copy Markdown
Member Author

I think it will be less confusing and inline with other distributions, to have 4 as default.

It might also delete people’s data unless they are aware of the bug and switch to transmission_3 before running transmission.

Can't find in all other threads, have you tried 4.0.4 ? I just tried downloading a torrent using 3, wait for it to start seeding. upgrade to 4.0.4 and file is still seeding for me

@doronbehar doronbehar left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Diff LGTM. CI is green.

@doronbehar
doronbehar merged commit 34585a2 into NixOS:master Oct 4, 2023
@kirillrdy
kirillrdy deleted the transmission branch October 4, 2023 18:57
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

6.topic: nixos Issues or PRs affecting NixOS modules, or package usability issues specific to NixOS 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 0 This PR does not cause any packages to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants