Skip to content

nodejs_22: 22.18.0 -> 22.19.0 - #438035

Merged
fabianhjr merged 1 commit into
NixOS:stagingfrom
aduh95:nodejs-22
Aug 29, 2025
Merged

nodejs_22: 22.18.0 -> 22.19.0#438035
fabianhjr merged 1 commit into
NixOS:stagingfrom
aduh95:nodejs-22

Conversation

@aduh95

@aduh95 aduh95 commented Aug 28, 2025

Copy link
Copy Markdown
Contributor

https://nodejs.org/en/blog/release/v22.19.0

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

Add a 👍 reaction to pull requests you find important.

@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 501+ This PR causes many rebuilds on Linux and should normally target the staging branches. 10.rebuild-darwin: 501+ This PR causes many rebuilds on Darwin and should normally target the staging branches. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 5001+ This PR causes many rebuilds on Darwin and must target the staging branches. 10.rebuild-linux: 5001+ This PR causes many rebuilds on Linux and must target the staging branches. 6.topic: nodejs Node.js is a free, open-source, cross-platform JavaScript runtime environment 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Aug 28, 2025
@aduh95
aduh95 requested a review from Conni2461 August 28, 2025 21:38
@nixpkgs-ci nixpkgs-ci Bot removed the 9.needs: reviewer This PR currently has no reviewers requested and needs attention. label Aug 28, 2025
@fabianhjr

Copy link
Copy Markdown
Member

Building with nix build github:aduh95/nixpkgs/nodejs-22#nodejs_22

@fabianhjr

fabianhjr commented Aug 29, 2025

Copy link
Copy Markdown
Member

Built on x86_64-linux

@fabianhjr
fabianhjr merged commit 2663ad9 into NixOS:staging Aug 29, 2025
36 of 38 checks passed
@aduh95
aduh95 deleted the nodejs-22 branch August 29, 2025 16:06
@nixpkgs-ci

nixpkgs-ci Bot commented Aug 29, 2025

Copy link
Copy Markdown
Contributor

Successfully created backport PR for staging-25.05:

@github-actions github-actions Bot added the 8.has: port to stable This PR already has a backport to the stable release. label Aug 29, 2025
@vcunat

vcunat commented Aug 31, 2025

Copy link
Copy Markdown
Member

I think this PR is why it fails on darwin now:
https://hydra.nixos.org/build/306435782

@vcunat

vcunat commented Aug 31, 2025

Copy link
Copy Markdown
Member

Actually, I suspect that something else must have caused it (but I don't know what).

@aduh95

aduh95 commented Aug 31, 2025

Copy link
Copy Markdown
Contributor Author

I've asked in #425699 (comment) as it could be related

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

6.topic: nodejs Node.js is a free, open-source, cross-platform JavaScript runtime environment 8.has: port to stable This PR already has a backport to the stable release. 10.rebuild-darwin: 501+ This PR causes many rebuilds on Darwin and should normally target the staging branches. 10.rebuild-darwin: 5001+ This PR causes many rebuilds on Darwin and must target the staging branches. 10.rebuild-linux: 501+ This PR causes many rebuilds on Linux and should normally target the staging branches. 10.rebuild-linux: 5001+ This PR causes many rebuilds on Linux and must target the staging branches. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants