ci: don't use the default problems handler in ci - #533355
Closed
jopejoe1 wants to merge 2 commits into
Closed
Conversation
Contributor
|
nixpkgs-branch-check
Bot
dismissed
their stale review
June 19, 2026 18:05
Review dismissed automatically
This makes it possible to use a custom broken massage or to mark a package as slated for removal without triggering the CI.
Member
Author
Missed the conditional for when we don't want to include broken packages in ci.
I don't see how unfree packages and broken packages and others don't block the channel either, so how would this be any diffrent?
If we want problem to be detected by ci we should add them like i did with broken here but unconditionally. |
13 tasks
Contributor
|
This was superseeded by #533376 |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This makes it possible to use a custom broken massage or to mark a package as slated for removal without triggering the CI.
closes #523712
Things done
passthru.tests.nixpkgs-reviewon this PR. See nixpkgs-review usage../result/bin/.