Skip to content

nodejs_24: 24.4.0 -> 24.4.1 - #425602

Merged
drupol merged 1 commit into
NixOS:masterfrom
aduh95:nodejs-24
Jul 18, 2025
Merged

nodejs_24: 24.4.0 -> 24.4.1#425602
drupol merged 1 commit into
NixOS:masterfrom
aduh95:nodejs-24

Conversation

@aduh95

@aduh95 aduh95 commented Jul 15, 2025

Copy link
Copy Markdown
Contributor

https://github.com/nodejs/node/releases/tag/v24.4.1

Things done

  • Built on platform(s)
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • For non-Linux: Is sandboxing enabled in nix.conf? (See Nix manual)
    • sandbox = relaxed
    • sandbox = true
  • Tested, as applicable:
  • Tested compilation of all packages that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage
  • Tested basic functionality of all binary files (usually in ./result/bin/)
  • Nixpkgs 25.11 Release Notes (or backporting 25.05 Nixpkgs Release notes)
    • (Package updates) Added a release notes entry if the change is major or breaking
  • NixOS 25.11 Release Notes (or backporting 25.05 NixOS Release notes)
    • (Module updates) Added a release notes entry if the change is significant
    • (Module addition) Added a release notes entry if adding a new NixOS module
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other contributing documentation in corresponding paths.

Add a 👍 reaction to pull requests you find important.

@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 6.topic: nodejs Node.js is a free, open-source, cross-platform JavaScript runtime environment labels Jul 15, 2025
@nixpkgs-ci nixpkgs-ci Bot added 9.needs: reviewer This PR currently has no reviewers requested and needs attention. and removed backport release-25.05 labels Jul 15, 2025
@aduh95
aduh95 requested a review from Conni2461 July 16, 2025 07:30
@nixpkgs-ci nixpkgs-ci Bot removed the 9.needs: reviewer This PR currently has no reviewers requested and needs attention. label Jul 16, 2025
@LordGrimmauld

Copy link
Copy Markdown
Contributor

CVE-2025-27209 probably affects us. CVE-2025-27210 is windows-only, so likely few people that are affected. Adding security tag.

@LordGrimmauld LordGrimmauld added the 1.severity: security Issues which raise a security issue, or PRs that fix one label Jul 16, 2025
@Conni2461

Copy link
Copy Markdown
Member

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 425602
Commit: d372ee5add1df609963e304e405aa517afbd3a77


x86_64-linux

❌ 1 package failed to build:
  • zipline
✅ 7 packages built:
  • corepack_24 (corepack_latest)
  • nodejs-slim_24 (nodejs-slim_latest)
  • nodejs-slim_24.dev (nodejs-slim_latest.dev)
  • nodejs-slim_24.libv8 (nodejs-slim_latest.libv8)
  • nodejs_24 (nodejs_latest)
  • nodejs_24.dev (nodejs_latest.dev)
  • nodejs_24.libv8 (nodejs_latest.libv8)

Error logs: `x86_64-linux`
zipline
Progress: resolved 1026, reused 979, downloaded 0, added 794
Progress: resolved 1026, reused 979, downloaded 0, added 806
Progress: resolved 1026, reused 979, downloaded 0, added 820
Progress: resolved 1026, reused 979, downloaded 0, added 831
Progress: resolved 1026, reused 979, downloaded 0, added 844
Progress: resolved 1026, reused 979, downloaded 0, added 855
Progress: resolved 1026, reused 979, downloaded 0, added 867
Progress: resolved 1026, reused 979, downloaded 0, added 878
Progress: resolved 1026, reused 979, downloaded 0, added 890
Progress: resolved 1026, reused 979, downloaded 0, added 897
Progress: resolved 1026, reused 979, downloaded 0, added 908
Progress: resolved 1026, reused 979, downloaded 0, added 920
Progress: resolved 1026, reused 979, downloaded 0, added 933
Progress: resolved 1026, reused 979, downloaded 0, added 945
Progress: resolved 1026, reused 979, downloaded 0, added 959
Progress: resolved 1026, reused 979, downloaded 0, added 971
Progress: resolved 1026, reused 979, downloaded 0, added 978
Progress: resolved 1026, reused 979, downloaded 0, added 979
 ERR_PNPM_NO_OFFLINE_TARBALL  A package is missing from the store but cannot download it in offline mode. The missing package may be downloaded from https://registry.npmjs.org/fast-xml-parser/-/fast-xml-parser-4.4.1.tgz.
Progress: resolved 1026, reused 979, downloaded 0, added 979

@nixpkgs-ci nixpkgs-ci Bot added the 12.approvals: 1 This PR was reviewed and approved by one person. label Jul 17, 2025
@aduh95

aduh95 commented Jul 18, 2025

Copy link
Copy Markdown
Contributor Author

The zipline failure looks unrelated. //cc @Defelo

Any chance this could be merged @dasJ @drupol

@drupol
drupol merged commit 910acd1 into NixOS:master Jul 18, 2025
3 of 4 checks passed
@nixpkgs-ci

nixpkgs-ci Bot commented Jul 18, 2025

Copy link
Copy Markdown
Contributor

Successfully created backport PR for release-25.05:

@github-actions github-actions Bot added the 8.has: port to stable This PR already has a backport to the stable release. label Jul 18, 2025
@aduh95
aduh95 deleted the nodejs-24 branch July 18, 2025 08:29
@0x4A6F

0x4A6F commented Jul 18, 2025

Copy link
Copy Markdown
Member

Could you also address the build failure on x86_64-darwin?

0x4A6F@darwin01 ~/nixpkgs (git)-[remotes/aduh95/nodejs-24] % nix-build --max-jobs 1 --cores 1 --system x86_64-darwin -A pkgs.nodejs_24
[...]
not ok 2573 parallel/test-runner-output
  ---
  duration_ms: 5193.87300
  severity: fail
  exitcode: 1
  stack: |-
    Test failure: 'test-runner/output/test-runner-watch-spec.mjs'
    Location: test/parallel/test-runner-output.mjs:346:5
    AssertionError [ERR_ASSERTION]: Expected values to be strictly equal:
    + actual - expected
    ... Skipped lines

      '✖ failing test (*ms)\n' +
        '✔ test has ran (*ms)\n' +
    +   'ℹ Error: A resource generated asynchronous activity after the test ended. This activity created the error "Error: EMFILE: too many open files, watch" which triggered an uncaughtException event, caught by the test runner.\n' +
        'ℹ tests 2\n' +
        'ℹ suites 0\n' +
        'ℹ pass 1\n' +
        'ℹ fail 1\n' +
        'ℹ cancelled 0\n' +
    ...
        '      *\n' +
    +   'ℹ Error: A resource generated asynchronous activity after the test ended. This activity created the error "Error: EMFILE: too many open files, watch" which triggered an uncaughtException event, caught by the test runner.\n' +
        'ℹ tests 0\n' +
        'ℹ suites 0\n' +
        'ℹ pass 0\n' +
        'ℹ fail 0\n' +
        'ℹ cancelled 0\n' +

        at assertSnapshot (/private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/common/assertSnapshot.js:63:12)
        at async Module.spawnAndAssert (/private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/common/assertSnapshot.js:98:3)
        at async TestContext.<anonymous> (file:///private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/parallel/test-runner-output.mjs:337:5)
        at async Test.run (node:internal/test_runner/test:1069:7)
        at async Promise.all (index 50)
        at async Suite.run (node:internal/test_runner/test:1466:7)
        at async startSubtestAfterBootstrap (node:internal/test_runner/harness:332:3) {
      generatedMessage: true,
      code: 'ERR_ASSERTION',
      actual: '✖ failing test (*ms)\n' +
        '✔ test has ran (*ms)\n' +
        'ℹ Error: A resource generated asynchronous activity after the test ended. This activity created the error "Error: EMFILE: too many open files, watch" which triggered an uncaughtException event, caught by the test runner.\n' +
        'ℹ tests 2\n' +
        'ℹ suites 0\n' +
        'ℹ pass 1\n' +
        'ℹ fail 1\n' +
        'ℹ cancelled 0\n' +
        'ℹ skipped 0\n' +
        'ℹ todo 0\n' +
        '...',
      expected: '✖ failing test (*ms)\n' +
        '✔ test has ran (*ms)\n' +
        'ℹ tests 2\n' +
        'ℹ suites 0\n' +
        'ℹ pass 1\n' +
        'ℹ fail 1\n' +
        'ℹ cancelled 0\n' +
        'ℹ skipped 0\n' +
        'ℹ todo 0\n' +
        'ℹ duration_ms *\n' +
        '...',
      operator: 'strictEqual'
    }

  ...
[...]
not ok 3053 parallel/test-tls-get-ca-certificates-system
  ---
  duration_ms: 262.00400
  severity: crashed
  exitcode: -11
  stack: |-
    ERROR: SecItemCopyMatching failed -50
  ...
not ok 3054 parallel/test-tls-get-ca-certificates-system-without-flag
  ---
  duration_ms: 304.46400
  severity: crashed
  exitcode: -11
  stack: |-
    ERROR: SecItemCopyMatching failed -50
  ...
[...]
not ok 3366 parallel/test-watch-file-shared-dependency
  ---
  duration_ms: 621.72700
  severity: fail
  exitcode: 1
  stack: |-
    Test failure: 'should not remove shared dependencies when unfiltering an owner'
    Location: test/parallel/test-watch-file-shared-dependency.mjs:33:3
    Error: EMFILE: too many open files, watch
        at FSWatcher._handle.onchange (node:internal/fs/watchers:214:21) {
      errno: -24,
      syscall: 'watch',
      code: 'EMFILE',
      filename: null
    }

  ...
[...]
not ok 3671 parallel/test-runner-complex-dependencies
  ---
  duration_ms: 120128.21000
  severity: fail
  exitcode: -15
  stack: |-
    timeout
  ...
[...]
not ok 3736 parallel/test-runner-global-setup-watch-mode
  ---
  duration_ms: 120095.52300
  severity: fail
  exitcode: -15
  stack: |-
    timeout
  ...
[...]
not ok 4226 sequential/test-watch-mode-watch-flags
  ---
  duration_ms: 512.71500
  severity: fail
  exitcode: 1
  stack: |-
    Test failure: 'when multiple `--watch` flags are provided should run as if only one was'
    Location: test/sequential/test-watch-mode-watch-flags.mjs:54:3
    AssertionError [ERR_ASSERTION]: Expected values to be strictly equal:
    + actual - expected

    + 'node:events:485\n' +
    +   "      throw er; // Unhandled 'error' event\n" +
    +   '      ^\n' +
    +   '\n' +
    +   'Error: EMFILE: too many open files, watch\n' +
    +   '    at FSWatcher._handle.onchange (node:internal/fs/watchers:214:21)\n' +
    +   "Emitted 'error' event on FSWatcher instance at:\n" +
    +   '    at FSWatcher._handle.onchange (node:internal/fs/watchers:220:12) {\n' +
    +   '  errno: -24,\n' +
    +   "  syscall: 'watch',\n" +
    +   "  code: 'EMFILE',\n" +
    +   '  filename: null\n' +
    +   '}\n' +
    +   '\n' +
    +   'Node.js v24.4.1\n'
    - ''

        at TestContext.<anonymous> (file:///private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/sequential/test-watch-mode-watch-flags.mjs:69:12)
        at process.processTicksAndRejections (node:internal/process/task_queues:105:5)
        at async Test.run (node:internal/test_runner/test:1069:7)
        at async Promise.all (index 0)
        at async Suite.run (node:internal/test_runner/test:1466:7)
        at async startSubtestAfterBootstrap (node:internal/test_runner/harness:332:3) {
      generatedMessage: true,
      code: 'ERR_ASSERTION',
      actual: 'node:events:485\n' +
        "      throw er; // Unhandled 'error' event\n" +
        '      ^\n' +
        '\n' +
        'Error: EMFILE: too many open files, watch\n' +
        '    at FSWatcher._handle.onchange (node:internal/fs/watchers:214:21)\n' +
        "Emitted 'error' event on FSWatcher instance at:\n" +
        '    at FSWatcher._handle.onchange (node:internal/fs/watchers:220:12) {\n' +
        '  errno: -24,\n' +
        "  syscall: 'watch',\n" +
        '...',
      expected: '',
      operator: 'strictEqual'
    }
  ...
[...]

Failed tests:
out/Release/node --expose-internals --test-reporter=./test/common/test-error-reporter.js --test-reporter-destination=stdout /private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/parallel/test-runner-output.mjs
out/Release/node --use-system-ca --test-reporter=./test/common/test-error-reporter.js --test-reporter-destination=stdout /private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/parallel/test-tls-get-ca-certificates-system.js
out/Release/node --test-reporter=./test/common/test-error-reporter.js --test-reporter-destination=stdout /private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/parallel/test-tls-get-ca-certificates-system-without-flag.js
out/Release/node --expose-internals --test-reporter=./test/common/test-error-reporter.js --test-reporter-destination=stdout /private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/parallel/test-watch-file-shared-dependency.mjs
out/Release/node --expose-internals --test-reporter=./test/common/test-error-reporter.js --test-reporter-destination=stdout /private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/parallel/test-runner-complex-dependencies.mjs
out/Release/node --test-reporter=./test/common/test-error-reporter.js --test-reporter-destination=stdout /private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/parallel/test-runner-global-setup-watch-mode.mjs
out/Release/node --test-reporter=./test/common/test-error-reporter.js --test-reporter-destination=stdout /private/tmp/nix-build-nodejs-24.4.1.drv-0/node-v24.4.1/test/sequential/test-watch-mode-watch-flags.mjs
make: *** [Makefile:589: test-ci-js] Error 1

@aduh95

aduh95 commented Jul 18, 2025

Copy link
Copy Markdown
Contributor Author

@0x4A6F these failures look sandbox-related, which should be addressed by #425699

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

1.severity: security Issues which raise a security issue, or PRs that fix one 6.topic: nodejs Node.js is a free, open-source, cross-platform JavaScript runtime environment 8.has: port to stable This PR already has a backport to the stable release. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 12.approvals: 1 This PR was reviewed and approved by one person.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants