Skip to content

[Backport release-25.05] nodejs_24: 24.4.0 -> 24.4.1 - #426316

Merged
drupol merged 1 commit into
release-25.05from
backport-425602-to-release-25.05
Jul 18, 2025
Merged

[Backport release-25.05] nodejs_24: 24.4.0 -> 24.4.1#426316
drupol merged 1 commit into
release-25.05from
backport-425602-to-release-25.05

Conversation

@nixpkgs-ci

@nixpkgs-ci nixpkgs-ci Bot commented Jul 18, 2025

Copy link
Copy Markdown
Contributor

Bot-based backport to release-25.05, triggered by a label in #425602.

  • Before merging, ensure that this backport is acceptable for the release.
    • Even as a non-committer, if you find that it is not acceptable, leave a comment.

(cherry picked from commit d372ee5)
@nixpkgs-ci nixpkgs-ci Bot added the 1.severity: security Issues which raise a security issue, or PRs that fix one label Jul 18, 2025
@nixpkgs-ci nixpkgs-ci Bot mentioned this pull request Jul 18, 2025
13 tasks
@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 6.topic: nodejs Node.js is a free, open-source, cross-platform JavaScript runtime environment 4.workflow: backport This targets a stable branch labels Jul 18, 2025
@nix-owners
nix-owners Bot requested a review from aduh95 July 18, 2025 08:33
@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. labels Jul 18, 2025
@aduh95

aduh95 commented Jul 18, 2025

Copy link
Copy Markdown
Contributor

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 426316
Commit: b527e89270879aaaf584c41f26b2796be634bc9d


aarch64-darwin

✅ 9 packages built:
  • corepack_24 (corepack_latest)
  • devmode
  • nodejs-slim_24 (nodejs-slim_latest)
  • nodejs-slim_24.dev (nodejs-slim_latest.dev)
  • nodejs-slim_24.libv8 (nodejs-slim_latest.libv8)
  • nodejs_24 (nodejs_latest)
  • nodejs_24.dev (nodejs_latest.dev)
  • nodejs_24.libv8 (nodejs_latest.libv8)
  • tailwindcss-language-server

@aduh95
aduh95 requested a review from drupol July 18, 2025 09:35
@drupol
drupol merged commit 97454bc into release-25.05 Jul 18, 2025
1 of 4 checks passed
@drupol
drupol deleted the backport-425602-to-release-25.05 branch July 18, 2025 09:39
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

1.severity: security Issues which raise a security issue, or PRs that fix one 4.workflow: backport This targets a stable branch 6.topic: nodejs Node.js is a free, open-source, cross-platform JavaScript runtime environment 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants